[release-4.19] OCPBUGS-92919,OCPBUGS-89936: Replace golang.org/x/crypto with github.com/openshift-sustaining/[email protected] - #10856
sunku5494 wants to merge 1 commit into
Conversation
…com/openshift-sustaining/[email protected] to address CVE-2026-39829, CVE-2026-46597 Signed-off-by: Venkata Charan Sunku <[email protected]>
|
Pipeline controller notification For optional jobs, comment This repository is configured in: LGTM mode |
|
Important Review skippedAuto reviews are disabled on base/target branches other than the default branch. Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Repository: openshift/coderabbit/.coderabbit.yaml Review profile: CHILL Plan: Enterprise Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
@sunku5494: This pull request references Jira Issue OCPBUGS-92919, which is invalid:
Comment The bug has been updated to refer to the pull request using the external bug tracker. This pull request references Jira Issue OCPBUGS-89936, which is invalid:
Comment The bug has been updated to refer to the pull request using the external bug tracker. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
/test images |
|
/label backport-risk-assessed |
|
Scheduling tests matching the |
|
/test e2e-aws-ovn |
1 similar comment
|
/test e2e-aws-ovn |
|
@sunku5494: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
Hello @sunku5494, I also see a similar PR #10840, which bumps to a (higher?) patch version Could you tell us which one we are going forward with? |
We are following a practice to choose the patch version that only fixes the attached CVE bug ids. due to this, we are seeing multiple PR's created with different patch version. At this time, we have seen a couple of them as there are many CVE's reported for crypto during the same period. So, I see this is an expected behavior and could consider this for merge. we will try to work on the ways to see if we could avoid this in future. Thanks |
address CVE-2026-39829, CVE-2026-46597