Skip to content

[main] Update Konflux configurations - #4144

Open
serverless-qe wants to merge 1 commit into
openshift-knative:mainfrom
serverless-qe:sync-konflux-main
Open

serverless-qe wants to merge 1 commit into
openshift-knative:mainfrom
serverless-qe:sync-konflux-main

Conversation

@serverless-qe

Copy link
Copy Markdown
Contributor

Update Konflux configurations

@Kaustubh-pande

Copy link
Copy Markdown
Contributor

/lgtm

@dsimansk

Copy link
Copy Markdown
Contributor

/approve
/lgtm

@dsimansk

Copy link
Copy Markdown
Contributor

/retest-required

@openshift-ci

openshift-ci Bot commented Aug 21, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dsimansk, Kaustubh-pande, serverless-qe

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [Kaustubh-pande,dsimansk]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@dsimansk

Copy link
Copy Markdown
Contributor

/retest

@openshift-ci

openshift-ci Bot commented Aug 24, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@serverless-qe
serverless-qe force-pushed the sync-konflux-main branch 3 times, most recently from a5ce247 to b991866 Compare August 26, 2026 14:05
@serverless-qe
serverless-qe force-pushed the sync-konflux-main branch 6 times, most recently from e28cf07 to 23af8d0 Compare September 1, 2026 22:45
@serverless-qe
serverless-qe force-pushed the sync-konflux-main branch 11 times, most recently from afb31eb to 3c2e633 Compare September 8, 2026 18:19
@serverless-qe
serverless-qe force-pushed the sync-konflux-main branch 3 times, most recently from d4f5de4 to 1e2b716 Compare September 15, 2026 11:47
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants