[release-v1.16] Update Konflux configurations - #1733
Open
serverless-qe wants to merge 2 commits into
Open
serverless-qe wants to merge 2 commits into
serverless-qe wants to merge 2 commits into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: serverless-qe The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
from
December 11, 2025 08:52
217c7bc to
fe26176
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
2 times, most recently
from
January 28, 2026 07:48
d807628 to
27446e1
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
from
February 2, 2026 20:05
27446e1 to
a1272fc
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
2 times, most recently
from
February 12, 2026 15:40
5ff45a6 to
e754ea7
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
10 times, most recently
from
March 5, 2026 13:28
688d460 to
de52cf5
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
6 times, most recently
from
March 12, 2026 13:49
387e83b to
e3fe405
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
5 times, most recently
from
March 18, 2026 06:20
4c91a5d to
fc728ef
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
3 times, most recently
from
April 1, 2026 12:59
6e88fef to
dd1bfaf
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
15 times, most recently
from
April 16, 2026 14:28
ae04b8f to
ae57771
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
9 times, most recently
from
April 27, 2026 07:58
80cc73f to
e4f8258
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
from
April 27, 2026 09:06
e4f8258 to
2708af7
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Update Konflux components and pipelines