Skip to content

deps(cargo): bump all policies - #262

Open
fslabs-bot[bot] wants to merge 52 commits into
mainfrom
updatecli_main_c6fe548f3a4dde3ec6ef0fe44ff3e92228049fd16df9e162f0aa0928369d8d10
Open

deps(cargo): bump all policies#262
fslabs-bot[bot] wants to merge 52 commits into
mainfrom
updatecli_main_c6fe548f3a4dde3ec6ef0fe44ff3e92228049fd16df9e162f0aa0928369d8d10

Conversation

@fslabs-bot

@fslabs-bot fslabs-bot Bot commented Oct 31, 2025

Copy link
Copy Markdown

deps(cargo): bump dependency "tracing-subscriber"

deps(cargo): bump dependency "tracing-subscriber" to 0.3.22

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.3.22\n"

deps(cargo): bump dependency "indicatif"

deps(cargo): bump dependency "indicatif" to 0.18.3

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.18.3\n"

deps(cargo): bump dependency "clap"

deps(cargo): bump dependency "clap" to 4.5.57

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 4.5.57\n"

deps(cargo): bump dependency "tokio"

deps(cargo): bump dependency "tokio" to 1.49.0

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 1.49.0\n"

deps(cargo): bump dependency "rustls"

deps(cargo): bump dependency "rustls" to 0.23.36

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.23.36\n"

deps(cargo): bump dependency "hyper-util"

deps(cargo): bump dependency "hyper-util" to 0.1.20

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.1.20\n"

deps(cargo): bump dependency "tempfile"

deps(cargo): bump dependency "tempfile" to 3.24.0

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 3.24.0\n"

deps(cargo): bump dependency "serde_with"

deps(cargo): bump dependency "serde_with" to 3.16.1

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 3.16.1\n"

deps(cargo): bump dependency "indexmap"

deps(cargo): bump dependency "indexmap" to 2.13.0

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 2.13.0\n"

deps(cargo): bump dependency "clap_complete"

deps(cargo): bump dependency "clap_complete" to 4.5.65

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 4.5.65\n"

deps(cargo): bump dependency "bytes"

deps(cargo): bump dependency "bytes" to 1.11.1

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 1.11.1\n"

deps(cargo): bump dependency "hyper"

deps(cargo): bump dependency "hyper" to 1.8.1

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 1.8.1\n"

deps(cargo): bump dependency "zip"

deps(cargo): bump dependency "zip" to 7.3.0

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 7.3.0\n"

deps(cargo): bump dependency "jsonwebtoken"

deps(cargo): bump dependency "jsonwebtoken" to 10.3.0

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 10.3.0\n"

deps(cargo): bump dev dependency "reqwest"

deps(cargo): bump dev dependency "reqwest" to 0.13.1

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.13.1\n"

deps(cargo): bump dependency "cargo-util"

deps(cargo): bump dependency "cargo-util" to 0.2.25

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.2.25\n"

deps(cargo): bump dependency "tracing-core"

deps(cargo): bump dependency "tracing-core" to 0.1.36

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.1.36\n"

deps(cargo): bump dependency "aws-config"

deps(cargo): bump dependency "aws-config" to 1.8.13

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 1.8.13\n"

deps(cargo): bump dependency "convert_case"

deps(cargo): bump dependency "convert_case" to 0.11.0

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.11.0\n"

deps(cargo): bump dependency "regex"

deps(cargo): bump dependency "regex" to 1.12.3

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 1.12.3\n"

deps(cargo): bump dependency "octocrab"

deps(cargo): bump dependency "octocrab" to 0.49.5

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.49.5\n"

deps(cargo): bump dependency "toml"

deps(cargo): bump dependency "toml" to 0.9.11+spec-1.1.0

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]+spec-1.1.0\ncargo update $ARGS --manifest-path Cargo.toml [email protected]+spec-1.1.0 --precise 0.9.11+spec-1.1.0\n"

deps(cargo): bump dependency "git2"

deps(cargo): bump dependency "git2" to 0.20.4

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.20.4\n"

deps(cargo): bump dependency "toml_edit"

deps(cargo): bump dependency "toml_edit" to 0.24.0+spec-1.1.0

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]+spec-1.1.0\ncargo update $ARGS --manifest-path Cargo.toml [email protected]+spec-1.1.0 --precise 0.24.0+spec-1.1.0\n"

deps(cargo): bump dependency "serde_json"

deps(cargo): bump dependency "serde_json" to 1.0.149

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 1.0.149\n"

deps(cargo): bump dependency "console"

deps(cargo): bump dependency "console" to 0.16.2

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.16.2\n"

deps(cargo): bump dependency "aws-sdk-cloudfront"

deps(cargo): bump dependency "aws-sdk-cloudfront" to 1.110.0

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 1.110.0\n"

deps(cargo): bump dev dependency "mockall"

deps(cargo): bump dev dependency "mockall" to 0.14.0

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.14.0\n"

deps(cargo): bump dependency "chrono"

deps(cargo): bump dependency "chrono" to 0.4.43

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.4.43\n"

deps(cargo): bump dependency "http"

deps(cargo): bump dependency "http" to 1.4.0

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 1.4.0\n"

deps(cargo): bump dependency "opendal"

deps(cargo): bump dependency "opendal" to 0.55.0

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.55.0\n"

deps(cargo): bump dependency "cargo_metadata"

deps(cargo): bump dependency "cargo_metadata" to 0.23.1

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.23.1\n"

deps(cargo): bump dependency "url"

deps(cargo): bump dependency "url" to 2.5.8

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 2.5.8\n"

deps(cargo): bump dev dependency "testcontainers"

deps(cargo): bump dev dependency "testcontainers" to 0.26.4

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.26.4\n"

deps(cargo): bump dependency "quick-xml"

deps(cargo): bump dependency "quick-xml" to 0.39.0

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.39.0\n"

deps(cargo): bump dependency "envconfig" for "fsl_test_api" crate

deps(cargo): bump crate dependency "envconfig" to 0.11.1

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path crates/fsl_test_api/Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path crates/fsl_test_api/Cargo.toml [email protected] --precise 0.11.1\n"

deps(cargo): bump dependency "tracing"

deps(cargo): bump dependency "tracing" to 0.1.44

ran shell command "ARGS=\"\"\nif [ \"$DRY_RUN\" = \"true\" ]; then\n ARGS=\"--dry-run\"\nfi\ncargo upgrade $ARGS --manifest-path Cargo.toml --package [email protected]\ncargo update $ARGS --manifest-path Cargo.toml [email protected] --precise 0.1.44\n"


Updatecli logo

Created automatically by Updatecli

Options:

Most of Updatecli configuration is done via its manifest(s).

  • If you close this pull request, Updatecli will automatically reopen it, the next time it runs.
  • If you close this pull request and delete the base branch, Updatecli will automatically recreate it, erasing all previous commits made.

Feel free to report any issues at github.com/updatecli/updatecli.
If you find this tool useful, do not hesitate to star our GitHub repository as a sign of appreciation, and/or to tell us directly on our chat!

@fslabs-bot fslabs-bot Bot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Oct 31, 2025
@fslabs-bot
fslabs-bot Bot force-pushed the updatecli_main_c6fe548f3a4dde3ec6ef0fe44ff3e92228049fd16df9e162f0aa0928369d8d10 branch 2 times, most recently from 678a50e to b902234 Compare October 31, 2025 19:02
@loispostula

Copy link
Copy Markdown
Member

stuck due to XAMPPRocky/octocrab#814

@fslabs-bot
fslabs-bot Bot force-pushed the updatecli_main_c6fe548f3a4dde3ec6ef0fe44ff3e92228049fd16df9e162f0aa0928369d8d10 branch 3 times, most recently from 19d495e to c97e2a2 Compare November 6, 2025 17:05
@fslabs-bot fslabs-bot Bot added size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. and removed size/M Denotes a PR that changes 30-99 lines, ignoring generated files. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Nov 6, 2025
@fslabs-bot
fslabs-bot Bot force-pushed the updatecli_main_c6fe548f3a4dde3ec6ef0fe44ff3e92228049fd16df9e162f0aa0928369d8d10 branch from 2694efc to 61596a2 Compare November 13, 2025 23:54
@fslabs-bot fslabs-bot Bot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. and removed size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Nov 15, 2025
@fslabs-bot
fslabs-bot Bot force-pushed the updatecli_main_c6fe548f3a4dde3ec6ef0fe44ff3e92228049fd16df9e162f0aa0928369d8d10 branch 2 times, most recently from 4d2c5cb to 4484169 Compare November 19, 2025 22:03
@fslabs-bot fslabs-bot Bot added size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. and removed size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Nov 20, 2025
@fslabs-bot
fslabs-bot Bot force-pushed the updatecli_main_c6fe548f3a4dde3ec6ef0fe44ff3e92228049fd16df9e162f0aa0928369d8d10 branch 7 times, most recently from c195a3e to b9cba27 Compare December 8, 2025 23:12
@fslabs-bot
fslabs-bot Bot force-pushed the updatecli_main_c6fe548f3a4dde3ec6ef0fe44ff3e92228049fd16df9e162f0aa0928369d8d10 branch from b7a9b99 to 43cf035 Compare December 10, 2025 20:31
@fslabs-bot fslabs-bot Bot added size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. and removed size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. labels Dec 17, 2025
@fslabs-bot
fslabs-bot Bot force-pushed the updatecli_main_c6fe548f3a4dde3ec6ef0fe44ff3e92228049fd16df9e162f0aa0928369d8d10 branch from fe1a400 to 902b595 Compare January 16, 2026 06:41
@fslabs-bot fslabs-bot Bot added size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. and removed size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels Jan 16, 2026
Made with ❤️️ by updatecli
@fslabs-bot fslabs-bot Bot removed the size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. label Jan 20, 2026
fmsc-bot added 4 commits January 22, 2026 14:01
@fslabs-bot

fslabs-bot Bot commented Jan 29, 2026

Copy link
Copy Markdown
Author

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign therawmeatball for approval. For more information see the Kubernetes Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/XL Denotes a PR that changes 500-999 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant