Skip to content

Update go toolchain to 1.26.5. - #1335

Merged
google-oss-prow[bot] merged 1 commit into
GoogleCloudPlatform:mainfrom
michelle192837:updates
Sep 2, 2026
Merged

Update go toolchain to 1.26.5.#1335
google-oss-prow[bot] merged 1 commit into
GoogleCloudPlatform:mainfrom
michelle192837:updates

Conversation

@michelle192837

Copy link
Copy Markdown
Collaborator

Previously updated the Go version to 1.26.5, but not the go toolchain. This should now match.

@google-oss-prow
google-oss-prow Bot requested a review from chases2 September 2, 2026 16:44
@michelle192837
michelle192837 removed the request for review from chases2 September 2, 2026 16:45
@chizhg

chizhg commented Sep 2, 2026

Copy link
Copy Markdown
Member

/lgtm

@google-oss-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: chizhg, michelle192837

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@google-oss-prow
google-oss-prow Bot merged commit f15bfa4 into GoogleCloudPlatform:main Sep 2, 2026
9 checks passed
@michelle192837
michelle192837 deleted the updates branch September 2, 2026 20:13
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants